launcher: print build info at startup #252

Merged
shizmob merged 3 commits from feature/launcher-version into master 2023-06-25 13:51:33 +03:00
shizmob commented 2023-05-01 11:44:58 +03:00 (Migrated from github.com)

Following the same pattern as inject and some hooks, useful for when getting debug logs.

One worry I have, same as when used elsewhere, is that version.c doesn't get rebuilt every build, allowing it to become stale. Does the Makefile system allow specifying a force-rebuild target?

Following the same pattern as `inject` and some hooks, useful for when getting debug logs. One worry I have, same as when used elsewhere, is that version.c doesn't get rebuilt every build, allowing it to become stale. Does the Makefile system allow specifying a force-rebuild target?
icex2 (Migrated from github.com) approved these changes 2023-05-01 17:13:51 +03:00
icex2 commented 2023-05-01 17:17:05 +03:00 (Migrated from github.com)

One worry I have, same as when used elsewhere, is that version.c doesn't get rebuilt every build, allowing it to become stale. Does the Makefile system allow specifying a force-rebuild target?

Not that I am aware of. The current way to do this is running clean builds whenever something gets actually distributed. So taking builds from our CI pipelines are fine, for local dev builds, you have to manually run a clean build. Not great, but the important stuff (official bin releases) works for now.

> One worry I have, same as when used elsewhere, is that version.c doesn't get rebuilt every build, allowing it to become stale. Does the Makefile system allow specifying a force-rebuild target? Not that I am aware of. The current way to do this is running clean builds whenever something gets actually distributed. So taking builds from our CI pipelines are fine, for local dev builds, you have to manually run a clean build. Not great, but the important stuff (official bin releases) works for now.
shizmob commented 2023-05-01 17:29:20 +03:00 (Migrated from github.com)

In that case, I just added that functionality. :)

In that case, I just added that functionality. :)
icex2 (Migrated from github.com) approved these changes 2023-05-01 17:36:04 +03:00
Sign in to join this conversation.